Update module gocv.io/x/gocv to v0.43.0 - #362
red-hat-konflux[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @red-hat-konflux[bot]. Thanks for your PR. I'm waiting for a opendatahub-io member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
254bf5c to
d29400d
Compare
|
d29400d to
6e8e319
Compare
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
6e8e319 to
7f3d293
Compare
This PR contains the following updates:
v0.35.0→v0.43.0Release Notes
hybridgroup/gocv (gocv.io/x/gocv)
v0.43.0: 0.43.0Compare Source
v0.42.0: 0.42.0Compare Source
v0.41.0: 0.41.0Compare Source
Dockerfile.opencv-ubuntu-24.04for Ubuntu 24.04 supportopencv-ubuntu-24-04step for build img workflowv0.40.0: 0.40.0Compare Source
v0.39.0: 0.39.0Compare Source
FaceDetectorYNexample.createLookUpTableandsplitfunctions.imgprocstandalone functions.XXXWithStreamstandalone functions.gocv-static-builderimage.opencvstatic.videoiofunctions to ROADMAP.objdetectfunctions for DNN faces, and movedarucounderobjdetectmodule.v0.38.0: 0.38.0Compare Source
v0.37.0: 0.37.0Compare Source
v0.36.1: 0.36.1Compare Source
v0.36.0: 0.36.0Compare Source
Feature2DalgorithmsFeature2DConfiguration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.