Skip to content

Update dependency tritonclient to v2.72.0 - #356

Open
red-hat-konflux[bot] wants to merge 1 commit into
konflux-pocfrom
konflux/mintmaker/konflux-poc/tritonclient-2.x
Open

red-hat-konflux[bot] wants to merge 1 commit into
konflux-pocfrom
konflux/mintmaker/konflux-poc/tritonclient-2.x

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Jun 17, 2026

Copy link
Copy Markdown

This PR contains the following updates:

Package Change Age Confidence
tritonclient ==2.51.0==2.72.0 age confidence
tritonclient ==2.45.0==2.72.0 age confidence
tritonclient ==2.37.0.9383150==2.72.0 age confidence
tritonclient ==2.41.0==2.72.0 age confidence
tritonclient ==2.44.0==2.72.0 age confidence
tritonclient <=2.45.0<=2.72.0 age confidence

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@openshift-ci

openshift-ci Bot commented Jun 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jun 17, 2026

Copy link
Copy Markdown

Hi @red-hat-konflux[bot]. Thanks for your PR.

I'm waiting for a opendatahub-io member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/konflux-poc/tritonclient-2.x branch from a997a2b to be39618 Compare June 29, 2026 18:43
@red-hat-konflux red-hat-konflux Bot changed the title Update dependency tritonclient to v2.69.0 Update dependency tritonclient to v2.70.0 Jun 29, 2026
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/konflux-poc/tritonclient-2.x branch from be39618 to ab51891 Compare August 2, 2026 01:08
@red-hat-konflux red-hat-konflux Bot changed the title Update dependency tritonclient to v2.70.0 Update dependency tritonclient to v2.71.0 Aug 2, 2026
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/konflux-poc/tritonclient-2.x branch from ab51891 to ebdd5f1 Compare September 6, 2026 01:23
@red-hat-konflux red-hat-konflux Bot changed the title Update dependency tritonclient to v2.71.0 Update dependency tritonclient to v2.72.0 Sep 6, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants